Skip to content

fix: respect targets in links - #2879

Merged
noahwaldner merged 3 commits into
mainfrom
fix/link-targets-FP-695
Oct 8, 2026
Merged

noahwaldner merged 3 commits into
mainfrom
fix/link-targets-FP-695

Conversation

@noahwaldner

Copy link
Copy Markdown
Contributor

When routed through the RouterProvider, any specified targets or held down keys get ignored.

This change uses browser routes instead of RouterProvider when a special target is requested, either by passing the target prop or holding down a key.

@noahwaldner
noahwaldner requested a review from syeo66 October 8, 2026 08:28
@noahwaldner
noahwaldner requested a review from a team as a code owner October 8, 2026 08:28
@changeset-bot

changeset-bot Bot commented Oct 8, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 82c0353

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 3 packages
Name Type
@frontify/fondue-components Patch
@frontify/fondue Patch
@frontify/fondue-sdk Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@syeo66 syeo66 left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

One small nit.

}

if (href) {
const useClientNavigation = !target || target === '_self';

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

use... is sort of a reserved word in React and should only be used for hooks.

@netlify

netlify Bot commented Oct 8, 2026

Copy link
Copy Markdown

✅ Deploy Preview for fondue-components ready!

Name Link
🔨 Latest commit 76e9442
🔍 Latest deploy log https://app.netlify.com/projects/fondue-components/deploys/6ac754423e8be80008ba800b
😎 Deploy Preview https://deploy-preview-2879.components.fondue-components.frontify.com
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify project configuration.

@netlify

netlify Bot commented Oct 8, 2026

Copy link
Copy Markdown

✅ Deploy Preview for fondue-components ready!

Name Link
🔨 Latest commit 82c0353
🔍 Latest deploy log https://app.netlify.com/projects/fondue-components/deploys/6ac7554f344bfe0007f9d9af
😎 Deploy Preview https://deploy-preview-2879.components.fondue-components.frontify.com
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify project configuration.

@noahwaldner
noahwaldner merged commit cff52fb into main Oct 8, 2026
19 checks passed
@noahwaldner
noahwaldner deleted the fix/link-targets-FP-695 branch October 8, 2026 08:40
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants