Skip to content

build(deps): bump crate-ci/typos from 1.50.3 to 1.51.1 - #3422

Merged
openshift-merge-bot[bot] merged 1 commit into
mainfrom
dependabot/github_actions/crate-ci/typos-1.51.1
Oct 9, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
mainfrom
dependabot/github_actions/crate-ci/typos-1.51.1

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Oct 9, 2026

Copy link
Copy Markdown
Contributor

Bumps crate-ci/typos from 1.50.3 to 1.51.1.

Release notes

Sourced from crate-ci/typos's releases.

v1.51.1

[1.51.1] - 2026-10-06

Features

  • Add ARM Windows builds

v1.51.0

[1.51.0] - 2026-10-06

Fixes

  • Don't correct bimap

Features

Changelog

Sourced from crate-ci/typos's changelog.

Change Log

All notable changes to this project will be documented in this file.

The format is based on Keep a Changelog and this project adheres to Semantic Versioning.

[Unreleased] - ReleaseDate

[1.51.1] - 2026-10-06

Features

  • Add ARM Windows builds

[1.51.0] - 2026-10-06

Fixes

  • Don't correct bimap

Features

[1.50.3] - 2026-09-25

Fixes

  • Don'y panicwhen case-converting non-ASCII corrections

[1.50.2] - 2026-09-15

Fixes

  • Don't panic when files being examined are removed

[1.50.1] - 2026-09-01

Fixes

  • Don't correct asend in Python code

[1.50.0] - 2026-08-28

Features

[1.49.1] - 2026-08-27

... (truncated)

Commits
  • 7f9d081 chore: Release
  • 4fc41d7 docs: Update changelog
  • eb4f299 Merge pull request #1602 from meop/ci/platform-release-matrix
  • 7623265 ci: release a win-arm64-msvc artifact
  • cf01e3f ci: match any windows runner when staging the release archive
  • dcea711 chore: Release
  • 73ce342 chore: Release
  • 58d970f docs: Update changelog
  • d59c567 Merge pull request #1634 from epage/bimap
  • 3689e3c fix(dict): Remove bimap from being corrected
  • Additional commits viewable in compare view

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [crate-ci/typos](https://github.com/crate-ci/typos) from 1.50.3 to 1.51.1.
- [Release notes](https://github.com/crate-ci/typos/releases)
- [Changelog](https://github.com/crate-ci/typos/blob/main/CHANGELOG.md)
- [Commits](crate-ci/typos@00f422f...7f9d081)

---
updated-dependencies:
- dependency-name: crate-ci/typos
  dependency-version: 1.51.1
  dependency-type: direct:production
  update-type: version-update:semver-minor
...

Signed-off-by: dependabot[bot] <support@github.com>
@openshift-ci

openshift-ci Bot commented Oct 9, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dependabot[bot], saschagrunert

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved label Oct 9, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit 336b5a1 into main Oct 9, 2026
34 checks passed
@dependabot
dependabot Bot deleted the dependabot/github_actions/crate-ci/typos-1.51.1 branch October 9, 2026 12:48
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant