Skip to content

chore: Lock file maintenance - #9145

Open
renovate[bot] wants to merge 1 commit into
mainfrom
renovate/lock-file-maintenance
Open

renovate[bot] wants to merge 1 commit into
mainfrom
renovate/lock-file-maintenance

Conversation

@renovate

@renovate renovate Bot commented Jun 16, 2025 •

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Update Change
lockFileMaintenance All locks refreshed

🔧 This Pull Request updates lock files to use the latest dependency versions.


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • Between 12:00 AM and 03:59 AM, only on Monday (* 0-3 * * 1)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@renovate
renovate Bot requested a review from a team as a code owner June 16, 2025 00:29
@vercel

vercel Bot commented Jun 16, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Actions Updated
blocksuite Error Error Oct 5, 2026 3:46pm UTC
1 Skipped Deployment
Project Deployment Actions Updated
blocksuite-docs Ignored Ignored Preview Oct 5, 2026 3:46pm UTC

@socket-security

Copy link
Copy Markdown

Review the following changes in direct dependencies. Learn more about Socket for GitHub.

Diff Package Supply Chain
Security
Vulnerability Quality Maintenance License
Addedclsx@​2.1.11001009576100

View full report

@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 7608a53 to 8bf6a22 Compare June 24, 2025 07:43
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 8bf6a22 to 8afc77d Compare July 1, 2025 05:48
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 8afc77d to 42e0115 Compare July 7, 2025 08:19
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 42e0115 to 3c4dd10 Compare August 10, 2025 14:03
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 3c4dd10 to 0de8a62 Compare August 13, 2025 14:58
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 0de8a62 to c441eb0 Compare August 19, 2025 18:14
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from c441eb0 to 7ba5f37 Compare August 31, 2025 13:00
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 7ba5f37 to 26ab247 Compare September 25, 2025 19:43
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 26ab247 to 2cd0440 Compare October 21, 2025 12:01
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 2cd0440 to b1d42a4 Compare November 10, 2025 16:59
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from b1d42a4 to e31d72e Compare November 18, 2025 22:12
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from e31d72e to e693011 Compare December 3, 2025 18:55
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 2e5d05d to f2257e2 Compare February 17, 2026 17:31
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from f2257e2 to eae0e69 Compare March 5, 2026 15:05
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from eae0e69 to 658b4cc Compare March 13, 2026 13:43
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 658b4cc to 7558a65 Compare April 1, 2026 20:41
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 7558a65 to 5cd4a4d Compare April 8, 2026 16:46
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 5cd4a4d to 1414a63 Compare April 29, 2026 14:19
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 1414a63 to d0d1fa9 Compare May 12, 2026 11:32
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from d0d1fa9 to 1386535 Compare May 18, 2026 18:39
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 1386535 to 90c7f48 Compare May 28, 2026 16:57
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 90c7f48 to d87b1c3 Compare June 1, 2026 18:47
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from d87b1c3 to e3ca28b Compare June 11, 2026 21:08
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from e3ca28b to 9020a5b Compare July 12, 2026 09:52
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 9020a5b to 8cf766c Compare July 17, 2026 00:25
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 8cf766c to 98839e2 Compare July 20, 2026 21:01
@renovate
renovate Bot force-pushed the renovate/lock-file-maintenance branch from 98839e2 to d6318cf Compare July 24, 2026 20:03

This branch had an error being deployed

2 failed (1 outdated) deployments
Preview — 9590114c Deployed Oct 5, 2026 by vercel[bot]
Preview – blocksuite — 5cd4a4d8 Deployed Apr 8, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

Status: No status

Development

Successfully merging this pull request may close these issues.

0 participants